Skip to content

chore(deps): bump the npm_and_yarn group across 1 directory with 2 updates - #2770

Open
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/npm_and_yarn-bd76ce8cdc
Open

dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/npm_and_yarn-bd76ce8cdc

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Oct 5, 2026

Copy link
Copy Markdown
Contributor

Bumps the npm_and_yarn group with 2 updates in the / directory: http-cache-semantics and webpack-dev-middleware.

Updates http-cache-semantics from 4.1.1 to 4.3.0

Commits

Updates webpack-dev-middleware from 7.4.2 to 7.4.6

Release notes

Sourced from webpack-dev-middleware's releases.

v7.4.6

7.4.6 (2026-09-03)

Bug Fixes

  • reject requests that resolve outside the output directory

v7.4.5

7.4.5 (2025-09-24)

Bug Fixes

v7.4.4

7.4.4 (2025-09-23)

Bug Fixes

v7.4.3

7.4.3 (2025-09-05)

Bug Fixes

  • do not call the next middleware for 304 responses (#2155) (c26a326)
  • do not call the next middleware when request is finished or errored (#2156) (116c680)
Changelog

Sourced from webpack-dev-middleware's changelog.

7.4.6 (2026-09-03)

Bug Fixes

  • reject requests that resolve outside the output directory

7.4.5 (2025-09-24)

Bug Fixes

7.4.4 (2025-09-23)

Bug Fixes

7.4.3 (2025-09-05)

Bug Fixes

  • do not call the next middleware for 304 responses (#2155) (c26a326)
  • do not call the next middleware when request is finished or errored (#2156) (116c680)
Commits
  • 70b3c33 chore(release): 7.4.6
  • cefccba Merge commit from fork
  • ec97e06 chore(release): 7.4.5
  • c9a0e68 fix: unpin memfs (#2176)
  • 0c87e95 chore(deps-dev): bump the dependencies group across 1 directory with 6 update...
  • 8f40f5f chore(release): 7.4.4
  • 044d691 fix: pin memfs version (#2174)
  • 253c7e1 chore(deps-dev): bump the dependencies group with 3 updates (#2170)
  • 2535262 chore(deps): bump the dependencies group across 1 directory with 5 updates (#...
  • 6b7531e chore: remove CODEOWNERS, too spammy (#2167)
  • Additional commits viewable in compare view

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore <dependency name> major version will close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)
  • @dependabot ignore <dependency name> minor version will close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)
  • @dependabot ignore <dependency name> will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)
  • @dependabot unignore <dependency name> will remove all of the ignore conditions of the specified dependency
  • @dependabot unignore <dependency name> <ignore condition> will remove the ignore condition of the specified dependency and ignore conditions
    You can disable automated security fix PRs for this repo from the Security Alerts page.

…dates

Bumps the npm_and_yarn group with 2 updates in the / directory: [http-cache-semantics](https://github.com/kornelski/http-cache-semantics) and [webpack-dev-middleware](https://github.com/webpack/webpack-dev-middleware).


Updates `http-cache-semantics` from 4.1.1 to 4.3.0
- [Commits](https://github.com/kornelski/http-cache-semantics/commits)

Updates `webpack-dev-middleware` from 7.4.2 to 7.4.6
- [Release notes](https://github.com/webpack/webpack-dev-middleware/releases)
- [Changelog](https://github.com/webpack/webpack-dev-middleware/blob/v7.4.6/CHANGELOG.md)
- [Commits](webpack/webpack-dev-middleware@v7.4.2...v7.4.6)

---
updated-dependencies:
- dependency-name: http-cache-semantics
  dependency-version: 4.3.0
  dependency-type: indirect
  dependency-group: npm_and_yarn
- dependency-name: webpack-dev-middleware
  dependency-version: 7.4.6
  dependency-type: indirect
  dependency-group: npm_and_yarn
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code labels Oct 5, 2026
@dependabot
dependabot Bot requested a review from a team as a code owner October 5, 2026 19:23
@dependabot
dependabot Bot removed the request for review from a team October 5, 2026 19:23
@dependabot dependabot Bot added the dependencies Pull requests that update a dependency file label Oct 5, 2026
@dependabot
dependabot Bot requested a review from mikeharv October 5, 2026 19:23
@dependabot dependabot Bot added the javascript Pull requests that update javascript code label Oct 5, 2026

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants