Skip to content

[main] Upgrade to latest dependencies - #16691

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
knative-automation:auto-updates/update-deps-main
Sep 21, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
knative-automation:auto-updates/update-deps-main

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

Cron -dprotaso

/cc knative/serving-writers
/assign knative/serving-writers

Produced by: knative-extensions/knobots/actions/update-deps

bumping knative.dev/networking 17b28c6...7d916c7:
  > 7d916c7 Fix non-contant format string err in go 1.26 (# 1152)
bumping knative.dev/pkg d2a153a...5dc1978:
  > 5dc1978 Bump google.golang.org/grpc from 1.83.0 to 1.83.2 (# 3389)
bumping google.golang.org/grpc 4c226da...030ee8b:
  > 030ee8b Update version to 1.83.2 (# 9375)
  > 8668b69 cherry-pick # 9365 to v1.83.x (# 9366)
  > a3e952d cherry-pick # 9346 to v1.83.x and update x/net dependency (# 9369)
  > 58f8fd9 Change version to 1.83.2-dev (# 9337)
  > 1550d9e Change version to 1.83.1 (# 9336)
  > ebba6f3 Cherry-pick # 9258 and # 9332 into v1.83.x (# 9335)
  > 8cfeca0 Cherry-pick # 9331 to v1.83.x (# 9333)
  > dec6951 Change version to 1.83.1-dev (# 9229)

Signed-off-by: Knative Automation <automation@knative.team>
@knative-prow
knative-prow Bot requested a review from a team September 21, 2026 01:39
@knative-prow knative-prow Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Sep 21, 2026
@knative-prow

knative-prow Bot commented Sep 21, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign dprotaso for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@codecov

codecov Bot commented Sep 21, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 80.36%. Comparing base (618b0c2) to head (022bd52).
⚠️ Report is 1 commits behind head on main.

Additional details and impacted files
@@            Coverage Diff             @@
##             main   #16691      +/-   ##
==========================================
+ Coverage   80.29%   80.36%   +0.07%     
==========================================
  Files         217      217              
  Lines       13560    13560              
==========================================
+ Hits        10888    10898      +10     
+ Misses       2309     2300       -9     
+ Partials      363      362       -1     

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@knative-prow
knative-prow Bot merged commit 017bbdb into knative:main Sep 21, 2026
67 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/L Denotes a PR that changes 100-499 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant